Skip to content

chore(deps): update konflux references - #367

Merged
michael-valdron merged 1 commit into
mainfrom
konflux/references/main
Jul 18, 2026
Merged

michael-valdron merged 1 commit into
mainfrom
konflux/references/main

Conversation

@red-hat-konflux

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Package Change Notes
quay.io/konflux-ci/tekton-catalog/task-apply-tags (source, changelog) a291081 → 6387614
quay.io/konflux-ci/tekton-catalog/task-build-image-index (source, changelog) b33bfa8 → 70c52e8
quay.io/konflux-ci/tekton-catalog/task-buildah-oci-ta (source, changelog) 85102e4 → daa9a28
quay.io/konflux-ci/tekton-catalog/task-clair-scan (source, changelog) 8fad4c2 → 9ff424d
quay.io/konflux-ci/tekton-catalog/task-clamav-scan (source, changelog) 567cb66 → 53a0232
quay.io/konflux-ci/tekton-catalog/task-ecosystem-cert-preflight-checks (source, changelog) 88f4fd6 → 2e5ebe0
quay.io/konflux-ci/tekton-catalog/task-git-clone-oci-ta (source, changelog) 0.1 → 0.2 ⚠️migration⚠️
quay.io/konflux-ci/tekton-catalog/task-prefetch-dependencies-oci-ta (source, changelog) 0e6324e → 92956e7
quay.io/konflux-ci/tekton-catalog/task-push-dockerfile-oci-ta (source, changelog) 7855471 → f3e97e6
quay.io/konflux-ci/tekton-catalog/task-rpms-signature-scan (source, changelog) 237c54b → 2e79e47
quay.io/konflux-ci/tekton-catalog/task-sast-shell-check-oci-ta (source, changelog) f960cc9 → f6a115e
quay.io/konflux-ci/tekton-catalog/task-sast-snyk-check-oci-ta (source, changelog) 0.4 → 0.5 ⚠️migration⚠️
quay.io/konflux-ci/tekton-catalog/task-sast-unicode-check-oci-ta (source, changelog) 324291c → 4961c44
quay.io/konflux-ci/tekton-catalog/task-show-sbom (source, changelog) a7346ed → 8fe70a9
quay.io/konflux-ci/tekton-catalog/task-source-build-oci-ta (source, changelog) 8567bb7 → 2dd5b3e

Release Notes

konflux-ci/build-definitions (quay.io/konflux-ci/tekton-catalog/task-git-clone-oci-ta)

v0.2


Configuration

📅 Schedule: (in timezone America/Toronto)

  • Branch creation
    • Between 05:00 AM and 11:59 PM, only on Saturday (* 5-23 * * 6)
  • Automerge
    • At any time (no schedule defined)

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

♻ Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.

👻 Immortal: This PR will be recreated if closed unmerged. Get config help if that's undesired.


  • If you want to rebase/retry this PR, check this box

To execute skipped test pipelines write comment /ok-to-test.


Documentation

Find out how to configure dependency updates in MintMaker documentation or see all available configuration options in Renovate documentation.

Signed-off-by: red-hat-konflux <126015336+red-hat-konflux[bot]@users.noreply.github.com>
@openshift-ci

openshift-ci Bot commented Jul 18, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: michael-valdron, red-hat-konflux[bot]

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@michael-valdron
michael-valdron merged commit 3f94c46 into main Jul 18, 2026
12 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant